Add @maxking to CODEOWNERS file by maxking · Pull Request #13599 · python/cpython · GitHub
Skip to content

Add @maxking to CODEOWNERS file - #13599

Merged
miss-islington merged 1 commit into
python:masterfrom
maxking:codeowner
May 28, 2019
Merged

Add @maxking to CODEOWNERS file#13599
miss-islington merged 1 commit into
python:masterfrom
maxking:codeowner

Conversation

@maxking

@maxking maxking commented May 27, 2019

Copy link
Copy Markdown
Contributor

No description provided.

@Mariatta

Copy link
Copy Markdown
Member

@Mariatta Mariatta changed the title Add myself to CODEOWNERS for email module. Add @maxking to CODEOWNERS file May 28, 2019
@miss-islington
miss-islington merged commit 71dc7c5 into python:master May 28, 2019
@maxking
maxking deleted the codeowner branch May 28, 2019 04:20
@maxking

maxking commented May 28, 2019

Copy link
Copy Markdown
Contributor Author

Thanks @Mariatta ! Now all I need to do is open a PR and see if it works ;-)

@tirkarthi

Copy link
Copy Markdown
Member

https://help.github.com/en/articles/about-code-owners

People with admin or owner permissions can set up a CODEOWNERS file in a repository. The people you choose as code owners must have write permissions for the repository.

brettcannon added a commit that referenced this pull request May 28, 2019
@brettcannon

Copy link
Copy Markdown
Member

DinoV pushed a commit to DinoV/cpython that referenced this pull request Jan 14, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

8 participants